Skip to content

chore(deps): update quay.io/rhacs-eng/konflux-tasks:latest docker digest to f2820c3 - #3772

Merged
red-hat-konflux[bot] merged 1 commit into
release-2.38from
konflux/mintmaker/release-2.38/stackrox-custom-konflux-tasks
Sep 6, 2026
Merged

chore(deps): update quay.io/rhacs-eng/konflux-tasks:latest docker digest to f2820c3#3772
red-hat-konflux[bot] merged 1 commit into
release-2.38from
konflux/mintmaker/release-2.38/stackrox-custom-konflux-tasks

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Sep 5, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
quay.io/rhacs-eng/konflux-tasks tekton-bundle digest 9ad31aaf2820c3

Configuration

📅 Schedule: (in timezone Etc/UTC)

  • Branch creation
    • Between 12:00 AM and 07:59 AM (* 0-7 * * *)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@red-hat-konflux
red-hat-konflux Bot requested review from a team and rhacs-bot as code owners September 5, 2026 07:43
@red-hat-konflux
red-hat-konflux Bot enabled auto-merge (squash) September 5, 2026 07:43

@rhacs-bot rhacs-bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Auto-approved by automation.

@github-actions

github-actions Bot commented Sep 5, 2026

Copy link
Copy Markdown

/retest scanner-db-slim-on-push

@openshift-ci

openshift-ci Bot commented Sep 5, 2026

Copy link
Copy Markdown

@github-actions[bot]: The /retest command does not accept any targets.
The following commands are available to trigger optional jobs:

/test e2e-tests
/test slim-e2e-tests

Use /test all to run all jobs.

Details

In response to this:

/retest scanner-db-slim-on-push

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@openshift-ci

openshift-ci Bot commented Sep 5, 2026

Copy link
Copy Markdown

@red-hat-konflux[bot]: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-tests 322a351 link false /test e2e-tests

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

…est to f2820c3

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux red-hat-konflux Bot changed the title chore(deps): update quay.io/rhacs-eng/konflux-tasks:latest docker digest to d7f5f96 chore(deps): update quay.io/rhacs-eng/konflux-tasks:latest docker digest to f2820c3 Sep 6, 2026
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/mintmaker/release-2.38/stackrox-custom-konflux-tasks branch from 322a351 to 3c3f16f Compare September 6, 2026 00:58
@red-hat-konflux
red-hat-konflux Bot merged commit df45763 into release-2.38 Sep 6, 2026
31 of 35 checks passed
@red-hat-konflux
red-hat-konflux Bot deleted the konflux/mintmaker/release-2.38/stackrox-custom-konflux-tasks branch September 6, 2026 01:11
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant