Skip to content

Bump Moq from 4.20.72 to 4.21.0 - #1929

Open
dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/nuget/Moq-4.21.0
Open

dependabot[bot] wants to merge 1 commit into
masterfrom
dependabot/nuget/Moq-4.21.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Oct 5, 2026 •

Copy link
Copy Markdown
Contributor

Updated Moq from 4.20.72 to 4.21.0.

Release notes

Sourced from Moq's releases.

4.21.0

What's Changed

✨ Implemented enhancements

🐛 Fixed bugs

🔨 Other

New Contributors

Full Changelog: devlooped/moq@v4.20.72...v4.21.0

Sponsors

The following sponsors made this release possible: @​clarius, @​MFB-Technologies-Inc, @​sandrock, @​drivenet, @​Keflon, @​tbolon, @​rbnswartz, @​jfoshee, @​Mrxx99, @​eajhnsn1, @​Jonathan-Hickey, @​KenBonny, @​SimonCropp, @​agileworks-eu, @​arsdragonfly, @​vezel-dev, @​ChilliCream, @​4OTC, @​DominicSchell, @​adalon, @​torutek, @​mccaffers, @​SeikaLogiciel, @​wizardness, @​eska-gmbh, @​geodata-no.

Thanks 💜

Commits viewable in compare view.

@dependabot dependabot Bot added the dependencies Pull requests that update a dependency file label Oct 5, 2026
@kubernetes-prow

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dependabot[bot]
Once this PR has been reviewed and has the lgtm label, please assign tg123 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Oct 5, 2026
@dependabot
dependabot Bot force-pushed the dependabot/nuget/Moq-4.21.0 branch from 6867cd5 to 0051432 Compare October 6, 2026 21:22
---
updated-dependencies:
- dependency-name: Moq
  dependency-version: 4.21.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/nuget/Moq-4.21.0 branch from 0051432 to 8ce9bc8 Compare October 6, 2026 22:14

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. dependencies Pull requests that update a dependency file size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants