Skip to content

CODEOWNERS: Remove jtojnar from discourse - #284

Merged
mweinelt merged 1 commit into
mainfrom
jtojnar-patch-1
Sep 1, 2026
Merged

CODEOWNERS: Remove jtojnar from discourse#284
mweinelt merged 1 commit into
mainfrom
jtojnar-patch-1

Conversation

@jtojnar

@jtojnar jtojnar commented Sep 1, 2026

Copy link
Copy Markdown
Member

Follow up to 9429575

@jtojnar
jtojnar requested a review from a team as a code owner September 1, 2026 22:30
@mweinelt
mweinelt merged commit fadbfb3 into main Sep 1, 2026
2 checks passed
@mweinelt
mweinelt deleted the jtojnar-patch-1 branch September 1, 2026 22:34
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants