Skip to content

[no-relnote] chore: fix typo in CODEOWNERS - #2056

Merged
cdesiniotis merged 1 commit into
NVIDIA:mainfrom
cdesiniotis:fix-typo-codeowners
Sep 9, 2026
Merged

[no-relnote] chore: fix typo in CODEOWNERS#2056
cdesiniotis merged 1 commit into
NVIDIA:mainfrom
cdesiniotis:fix-typo-codeowners

Conversation

@cdesiniotis

Copy link
Copy Markdown
Contributor

No description provided.

Signed-off-by: Christopher Desiniotis <cdesiniotis@nvidia.com>
@cdesiniotis cdesiniotis self-assigned this Sep 8, 2026
@coveralls

Copy link
Copy Markdown

Coverage Report for CI Build 34288046231

Coverage remained the same at 44.017%

Details

  • Coverage remained the same as the base build.
  • Patch coverage: No coverable lines changed in this PR.
  • No coverage regressions found.

Uncovered Changes

No uncovered changes found.

Coverage Regressions

No coverage regressions found.


Coverage Stats

Coverage Status
Relevant Lines: 13522
Covered Lines: 5952
Line Coverage: 44.02%
Coverage Strength: 0.44 hits per line

💛 - Coveralls

@cdesiniotis
cdesiniotis merged commit 4165c0e into NVIDIA:main Sep 9, 2026
13 of 20 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants